Skip to content

Use referenced resource name for suffix-only reference fields - #746

Open
visit1985 wants to merge 1 commit into
aws-controllers-k8s:mainfrom
visit1985:feature/suffix-only-references
Open

visit1985 wants to merge 1 commit into
aws-controllers-k8s:mainfrom
visit1985:feature/suffix-only-references

Conversation

@visit1985

Copy link
Copy Markdown
Contributor

Issue #, if available:

aws-controllers-k8s/community#2173
As suggested in #721 (comment)

Description of changes:

Use the referenced resource name as the reference-field prefix when the source field name consists of an identifier suffix only and added WAFv2 test coverage for this.

By submitting this pull request, I confirm that my contribution is made under the terms of the Apache 2.0 license.

@ack-prow
ack-prow Bot requested review from jlbutler and knottnt September 30, 2026 20:01
@ack-prow

ack-prow Bot commented Sep 30, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: visit1985
Once this PR has been reviewed and has the lgtm label, please assign michaelhtm for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ack-prow ack-prow Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Sep 30, 2026
@ack-prow

ack-prow Bot commented Sep 30, 2026

Copy link
Copy Markdown

Hi @visit1985. Thanks for your PR.

I'm waiting for a aws-controllers-k8s member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Tip

We noticed you've done this a few times! Consider joining the org to skip this step and gain /lgtm and other bot rights. We recommend asking approvers on your previous PRs to sponsor you.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant