Skip to content

fix: report adoption/readonly not found errors in condition - #270

Open
michaelhtm wants to merge 1 commit into
aws-controllers-k8s:mainfrom
michaelhtm:fix/reportadoptionnotfounderror
Open

michaelhtm wants to merge 1 commit into
aws-controllers-k8s:mainfrom
michaelhtm:fix/reportadoptionnotfounderror

Conversation

@michaelhtm

Copy link
Copy Markdown
Member

By submitting this pull request, I confirm that my contribution is made under the terms of the Apache 2.0 license.

@ack-prow
ack-prow Bot requested review from a-hilaly and knottnt September 18, 2026 20:58
@ack-prow

ack-prow Bot commented Sep 18, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: michaelhtm

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ack-prow ack-prow Bot added the approved label Sep 18, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant