Skip to content

[main] Source code updates from microsoft/vstest - #9254

Merged
dotnet-maestro[bot] merged 2 commits into
mainfrom
darc-main-a256baa6-2b6d-40b7-bf70-3f1352a39d6d
Sep 11, 2026
Merged

[main] Source code updates from microsoft/vstest#9254
dotnet-maestro[bot] merged 2 commits into
mainfrom
darc-main-a256baa6-2b6d-40b7-bf70-3f1352a39d6d

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/microsoft/vstest

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/microsoft/vstest:5b51e21b44706edf1bba8694c25a546d27619f2c..https://github.com/dotnet/dotnet:darc-main-a256baa6-2b6d-40b7-bf70-3f1352a39d6d

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Important

⚠️ No subscription flowing code in the opposite direction is set up.
Code changes made in the VMR's src/ directory will not flow back to the source repository on branch "main".

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 1 pipeline(s).
1 pipeline(s) were filtered out due to trigger conditions.
There may be pipelines that require an authorized user to comment /azp run to run.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team September 11, 2026 02:05

@maestro-approver maestro-approver Bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This pull request contains only source updates from https://github.com/microsoft/vstest.

@dotnet-maestro
dotnet-maestro Bot merged commit fd98104 into main Sep 11, 2026
16 checks passed
@dotnet-maestro
dotnet-maestro Bot deleted the darc-main-a256baa6-2b6d-40b7-bf70-3f1352a39d6d branch September 11, 2026 04:25
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants