Skip to content

OpenAPI schema sync (v3.8.0) - #291

Merged
mcnulty-fp merged 2 commits into
mainfrom
feat/open-api-v3.8.0
Sep 28, 2026
Merged

mcnulty-fp merged 2 commits into
mainfrom
feat/open-api-v3.8.0

Conversation

@fingerprint-dx-team-actions-runner

@fingerprint-dx-team-actions-runner fingerprint-dx-team-actions-runner Bot commented Sep 24, 2026 •

Copy link
Copy Markdown
Contributor

Schema sync for v3.8.0 OpenAPI schema release.

Tracked with INTER-2584

@changeset-bot

changeset-bot Bot commented Sep 24, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 68eb49f

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Sep 24, 2026 •

Copy link
Copy Markdown
Contributor

Coverage report

St.❔
Category Percentage Covered / Total
🟢 Statements 98.25% 168/171
🟢 Branches 97.03% 98/101
🟢 Functions 100% 37/37
🟢 Lines 98.21% 165/168

Test suite run success

132 tests passing in 30 suites.

Report generated by 🧪jest coverage report action from 68eb49f

Show full coverage report
St File % Stmts % Branch % Funcs % Lines Uncovered Line #s
🟢 All files 98.24 97.02 100 98.21
🟢  src 98.54 98.76 100 98.51
🔴   ...edApiTypes.ts 0 0 0 0
🔴   index.ts 0 0 0 0
🟢   sealedResults.ts 100 100 100 100
🟢   ...rApiClient.ts 96.15 97.22 100 96.15 360,364
🟢   types.ts 100 100 100 100
🟢   urlUtils.ts 100 100 100 100
🟢   webhook.ts 100 100 100 100
🟢  src/errors 97.05 90 100 96.96
🟢   apiErrors.ts 100 100 100 100
🟢   ...orResponse.ts 100 100 100 100
🟢   toError.ts 87.5 88.88 100 87.5 21
🟢   unsealError.ts 100 50 100 100 14

@github-actions

Copy link
Copy Markdown
Contributor

🚀 Following releases will be created using changesets from this PR:

@fingerprint/node-sdk@7.8.0

Minor Changes

  • events: Add device_details smart signal to Event (0dda9e8)

Patch Changes

  • Throw SdkError instead of TypeError or Error for invalid API client arguments. (fde9447)
  • events: Fix descriptions to use ID rather than Id. Clarify descriptions for Labels. (0dda9e8)

@mcnulty-fp
mcnulty-fp merged commit 7f78656 into main Sep 28, 2026
19 checks passed
@mcnulty-fp
mcnulty-fp deleted the feat/open-api-v3.8.0 branch September 28, 2026 13:29
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants