Skip to content

Add CODEOWNERS file - #2451

Merged
willdurand merged 1 commit into
masterfrom
codeowners
Sep 15, 2026
Merged

willdurand merged 1 commit into
masterfrom
codeowners

Conversation

@willdurand

Copy link
Copy Markdown
Member

No description provided.

@willdurand
willdurand requested a review from diox September 15, 2026 07:31
@willdurand
willdurand merged commit 738e929 into master Sep 15, 2026
5 checks passed
@willdurand
willdurand deleted the codeowners branch September 15, 2026 12:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants