Skip to content

Bump github.com/onsi/ginkgo/v2 from 2.32.0 to 2.32.1 in the test-dependencies group - #28

Merged
acornett21 merged 1 commit into
mainfrom
dependabot/go_modules/test-dependencies-dd0f7adeb4
Aug 21, 2026
Merged

Bump github.com/onsi/ginkgo/v2 from 2.32.0 to 2.32.1 in the test-dependencies group#28
acornett21 merged 1 commit into
mainfrom
dependabot/go_modules/test-dependencies-dd0f7adeb4

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Aug 20, 2026

Copy link
Copy Markdown
Contributor

Bumps the test-dependencies group with 1 update: github.com/onsi/ginkgo/v2.

Updates github.com/onsi/ginkgo/v2 from 2.32.0 to 2.32.1

Release notes

Sourced from github.com/onsi/ginkgo/v2's releases.

v2.32.1

2.32.1

Fixes

  • Defer AfterAll until repeated spec completes [e647b3b]
Changelog

Sourced from github.com/onsi/ginkgo/v2's changelog.

2.32.1

Fixes

  • Defer AfterAll until repeated spec completes [e647b3b]
Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore <dependency name> major version will close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)
  • @dependabot ignore <dependency name> minor version will close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)
  • @dependabot ignore <dependency name> will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)
  • @dependabot unignore <dependency name> will remove all of the ignore conditions of the specified dependency
  • @dependabot unignore <dependency name> <ignore condition> will remove the ignore condition of the specified dependency and ignore conditions

Bumps the test-dependencies group with 1 update: [github.com/onsi/ginkgo/v2](https://github.com/onsi/ginkgo).


Updates `github.com/onsi/ginkgo/v2` from 2.32.0 to 2.32.1
- [Release notes](https://github.com/onsi/ginkgo/releases)
- [Changelog](https://github.com/onsi/ginkgo/blob/master/CHANGELOG.md)
- [Commits](onsi/ginkgo@v2.32.0...v2.32.1)

---
updated-dependencies:
- dependency-name: github.com/onsi/ginkgo/v2
  dependency-version: 2.32.1
  dependency-type: direct:production
  update-type: version-update:semver-patch
  dependency-group: test-dependencies
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file go Pull requests that update go code labels Aug 20, 2026
@exe-prow-github-app

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: acornett21, dependabot[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@exe-prow-github-app exe-prow-github-app Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 21, 2026
@acornett21
acornett21 merged commit 265350f into main Aug 21, 2026
1 of 2 checks passed
@dependabot
dependabot Bot deleted the dependabot/go_modules/test-dependencies-dd0f7adeb4 branch August 21, 2026 20:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dependencies Pull requests that update a dependency file go Pull requests that update go code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant