Skip to content

Catalog Update from Bundle Snapshot crt-nshift-lightspeed-tenant/ols-bundle-20260908-183121-000 - #2035

Open
raptorsun wants to merge 1 commit into
openshift:mainfrom
raptorsun:fbc-update
Open

raptorsun wants to merge 1 commit into
openshift:mainfrom
raptorsun:fbc-update

Conversation

@raptorsun

Copy link
Copy Markdown
Contributor

This PR is triggered by the release crt-nshift-lightspeed-tenant/bundle-release-20260908-183942.

@coderabbitai

coderabbitai Bot commented Sep 8, 2026

Copy link
Copy Markdown

Comment @coderabbitai help to get the list of available commands.

@red-hat-konflux red-hat-konflux Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. labels Sep 8, 2026
@red-hat-konflux

Copy link
Copy Markdown
Contributor

label added: lgtm, approved

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by:

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 0b9ff41 and 2 for PR HEAD 2143c56 in total

@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

@raptorsun: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@sriroopar

Copy link
Copy Markdown
Contributor

/retest

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants